Skip to content

fix: keep selected options in search multiselect - #1032

Open
sagardspeed2 wants to merge 11 commits into
mainfrom
1026-keep-selected-options-in-search-multiselect
Open

sagardspeed2 wants to merge 11 commits into
mainfrom
1026-keep-selected-options-in-search-multiselect

Conversation

@sagardspeed2

@sagardspeed2 sagardspeed2 commented Feb 8, 2025 •

Copy link
Copy Markdown
Contributor

Describe your changes

keep selected options in search multiselect

Screenshots [Optional]

Issue ticket number and link

Closes #1028

Checklist before requesting a review

  • I have performed a self-review of my code
  • I have added tests
  • I have added a changeset pnpm changeset add
  • I have added example usage in the kitchen sink app

@sagardspeed2 sagardspeed2 self-assigned this Feb 8, 2025
@changeset-bot

changeset-bot Bot commented Feb 8, 2025 •

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: dc04bd1

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 4 packages
Name Type
@ensembleui/react-kitchen-sink Patch
@ensembleui/react-runtime Patch
@ensembleui/react-preview Patch
@ensembleui/react-starter Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

github-actions Bot commented Feb 8, 2025 •

Copy link
Copy Markdown
Contributor

Visit the preview URL for this PR (updated for commit 812fcd5):

https://react-kitchen-sink-dev--pr1032-1026-keep-selected-o-ladrvyna.web.app

(expires Sat, 29 Mar 2025 07:28:09 GMT)

🔥 via Firebase Hosting GitHub Action 🌎

Sign: 6267897ade2ba783b6db70a53a60fc3946d625e9

Comment thread packages/runtime/src/widgets/Form/MultiSelect.tsx Outdated
@kmahmood74
kmahmood74 force-pushed the 1026-keep-selected-options-in-search-multiselect branch from 812fcd5 to dc04bd1 Compare October 1, 2026 23:00
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants