Skip to content

fix(immutable): upgrade to v5 (option 1) - #5743

Open
smouillour wants to merge 33 commits into
masterfrom
smouillour-migrate-immutable-5
Open

smouillour wants to merge 33 commits into
masterfrom
smouillour-migrate-immutable-5

Conversation

@smouillour

Copy link
Copy Markdown
Contributor

What is the problem this PR is trying to solve?

What is the chosen solution to this problem?

Please check if the PR fulfills these requirements

  • The PR have used yarn changeset to a request a release from the CI if wanted.
  • The PR commit message follows our guidelines
  • Tests for the changes have been added (for bug fixes / features) And non reg done before need review
  • Docs have been added / updated (for bug fixes / features)
  • Related design / discussions / pages (not in jira), if any, are all linked or available in the PR

[ ] This PR introduces a breaking change

@changeset-bot

changeset-bot Bot commented Mar 9, 2026 •

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: d815ecf

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 13 packages
Name Type
@talend/react-cmf-cqrs Major
@talend/react-cmf Major
@talend/react-components Major
@talend/react-containers Major
@talend/react-flow-designer Major
@talend/react-sagas Major
@talend/react-cmf-router Patch
@talend/react-storybook-cmf Patch
@talend/scripts-config-storybook-lib Patch
@talend/react-dataviz Patch
@talend/react-faceted-search Patch
@talend/react-forms Patch
@talend/react-stepper Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@smouillour smouillour changed the title fix(immutable): upgrade to v5 fix(immutable): upgrade to v5 (option 1) Mar 9, 2026
@smouillour
smouillour force-pushed the smouillour-migrate-immutable-5 branch from 45b2cd6 to 6fff7c8 Compare March 9, 2026 12:33
@smouillour
smouillour temporarily deployed to pull_request_unsafe March 9, 2026 12:34 — with GitHub Actions Inactive
@smouillour
smouillour temporarily deployed to pull_request_unsafe March 9, 2026 12:34 — with GitHub Actions Inactive
@smouillour
smouillour temporarily deployed to pull_request_unsafe March 9, 2026 13:30 — with GitHub Actions Inactive
@smouillour
smouillour temporarily deployed to pull_request_unsafe March 9, 2026 13:30 — with GitHub Actions Inactive
@smouillour
smouillour had a problem deploying to pull_request_unsafe March 9, 2026 13:30 — with GitHub Actions Failure
@smouillour
smouillour had a problem deploying to pull_request_unsafe March 9, 2026 13:30 — with GitHub Actions Failure
@smouillour
smouillour temporarily deployed to pull_request_unsafe March 9, 2026 13:43 — with GitHub Actions Inactive
@smouillour
smouillour temporarily deployed to pull_request_unsafe March 9, 2026 13:43 — with GitHub Actions Inactive
@smouillour
smouillour temporarily deployed to pull_request_unsafe March 9, 2026 13:43 — with GitHub Actions Inactive
@github-actions

github-actions Bot commented Mar 9, 2026

Copy link
Copy Markdown
Contributor

Storybook for this PR deployed on this github page

@smouillour
smouillour temporarily deployed to pull_request_unsafe March 9, 2026 13:53 — with GitHub Actions Inactive
@smouillour
smouillour temporarily deployed to pull_request_unsafe March 9, 2026 13:53 — with GitHub Actions Inactive
@smouillour
smouillour temporarily deployed to pull_request_unsafe March 9, 2026 13:53 — with GitHub Actions Inactive
@smouillour
smouillour had a problem deploying to pull_request_unsafe March 9, 2026 13:53 — with GitHub Actions Failure
@smouillour
smouillour marked this pull request as ready for review March 9, 2026 14:01
@smouillour
smouillour temporarily deployed to pull_request_unsafe March 9, 2026 15:01 — with GitHub Actions Inactive
@smouillour
smouillour temporarily deployed to pull_request_unsafe March 9, 2026 15:01 — with GitHub Actions Inactive
@smouillour
smouillour temporarily deployed to pull_request_unsafe March 9, 2026 15:01 — with GitHub Actions Inactive
@smouillour
smouillour temporarily deployed to pull_request_unsafe March 9, 2026 15:01 — with GitHub Actions Inactive
@jmfrancois
jmfrancois deployed to pull_request_unsafe September 25, 2026 08:34 — with GitHub Actions Active
@jmfrancois
jmfrancois deployed to pull_request_unsafe September 25, 2026 08:34 — with GitHub Actions Active

This branch had an error being deployed

1 failed deployment
pull_request_unsafe — d815ecf6 Deployed Sep 25, 2026 by jmfrancois via Build and test #8369
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants