Skip to content

Update CODEOWNERS - #918

Merged
quetzalliwrites merged 1 commit into
mainfrom
codeowners-auto-updates-docs
Sep 7, 2026
Merged

Update CODEOWNERS#918
quetzalliwrites merged 1 commit into
mainfrom
codeowners-auto-updates-docs

Conversation

@localstack-bot

Copy link
Copy Markdown
Collaborator

Automatic update of CODEOWNERS file.

@cloudflare-workers-and-pages

cloudflare-workers-and-pages Bot commented Sep 7, 2026

Copy link
Copy Markdown

Deploying localstack-docs with  Cloudflare Pages  Cloudflare Pages

Latest commit: 18f14ee
Status: ✅  Deploy successful!
Preview URL: https://687a7ace.localstack-docs.pages.dev
Branch Preview URL: https://codeowners-auto-updates-docs.localstack-docs.pages.dev

View logs

@quetzalliwrites
quetzalliwrites merged commit 831036b into main Sep 7, 2026
1 of 2 checks passed
@quetzalliwrites
quetzalliwrites deleted the codeowners-auto-updates-docs branch September 7, 2026 07:07
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants