Skip to content

OpenSSL pin revert - #4044

Draft
michburk wants to merge 2 commits into
openstack-k8s-operators:mainfrom
michburk:ssl-pin-revert
Draft

michburk wants to merge 2 commits into
openstack-k8s-operators:mainfrom
michburk:ssl-pin-revert

Conversation

@michburk

@michburk michburk commented Jul 8, 2026

Copy link
Copy Markdown
Contributor

Check jobs on this pr should pass once a newer, non-broken version is available

@openshift-ci

openshift-ci Bot commented Jul 8, 2026

Copy link
Copy Markdown
Contributor

Skipping CI for Draft Pull Request.
If you want CI signal for your change, please convert it to an actual PR.
You can still manually trigger a test run with /test all

@openshift-ci

openshift-ci Bot commented Jul 8, 2026

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by:
Once this PR has been reviewed and has the lgtm label, please assign fultonj for approval. For more information see the Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@michburk michburk changed the title Ssl pin revert OpenSSL pin revert Jul 8, 2026
@centosinfra-prod-github-app

Copy link
Copy Markdown

Build failed (check pipeline). Post recheck (without leading slash)
to rerun all jobs. Make sure the failure cause has been resolved before
you rerun jobs.

https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/b95a1789a2a04413af48fd1496f83a84

❌ openstack-k8s-operators-content-provider FAILURE in 5m 19s
⚠️ podified-multinode-edpm-deployment-crc SKIPPED Skipped due to failed job openstack-k8s-operators-content-provider
⚠️ cifmw-crc-podified-edpm-baremetal SKIPPED Skipped due to failed job openstack-k8s-operators-content-provider
⚠️ cifmw-crc-podified-edpm-baremetal-minor-update SKIPPED Skipped due to failed job openstack-k8s-operators-content-provider
⚠️ podified-multinode-hci-deployment-crc SKIPPED Skipped due to failed job openstack-k8s-operators-content-provider
⚠️ cifmw-multinode-tempest SKIPPED Skipped due to failed job openstack-k8s-operators-content-provider
⚠️ adoption-standalone-to-crc-ceph-provider SKIPPED Skipped due to failed job openstack-k8s-operators-content-provider
⚠️ whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp SKIPPED Skipped due to failed job openstack-k8s-operators-content-provider
✔️ noop SUCCESS in 0s
❌ cifmw-pod-ansible-test FAILURE in 6m 38s
✔️ cifmw-pod-pre-commit SUCCESS in 8m 09s
❌ cifmw-content-provider-build-images FAILURE in 4m 47s
❌ cifmw-multinode-kuttl FAILURE in 21m 42s
❌ cifmw-tcib FAILURE in 5m 18s
❌ cifmw-architecture-validate-hci FAILURE in 3m 52s
❌ ci-framework-openstack-meta-content-provider FAILURE in 4m 40s

@centosinfra-prod-github-app

Copy link
Copy Markdown

Build failed (check pipeline). Post recheck (without leading slash)
to rerun all jobs. Make sure the failure cause has been resolved before
you rerun jobs.

https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/e2cf30ff02884d23bf2099a2205b786a

❌ openstack-k8s-operators-content-provider FAILURE in 4m 38s
⚠️ podified-multinode-edpm-deployment-crc SKIPPED Skipped due to failed job openstack-k8s-operators-content-provider
⚠️ cifmw-crc-podified-edpm-baremetal SKIPPED Skipped due to failed job openstack-k8s-operators-content-provider
⚠️ cifmw-crc-podified-edpm-baremetal-minor-update SKIPPED Skipped due to failed job openstack-k8s-operators-content-provider
⚠️ podified-multinode-hci-deployment-crc SKIPPED Skipped due to failed job openstack-k8s-operators-content-provider
⚠️ cifmw-multinode-tempest SKIPPED Skipped due to failed job openstack-k8s-operators-content-provider
⚠️ adoption-standalone-to-crc-ceph-provider SKIPPED Skipped due to failed job openstack-k8s-operators-content-provider
⚠️ whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp SKIPPED Skipped due to failed job openstack-k8s-operators-content-provider
✔️ noop SUCCESS in 0s
❌ cifmw-pod-ansible-test FAILURE in 6m 14s
✔️ cifmw-pod-pre-commit SUCCESS in 8m 40s
❌ cifmw-content-provider-build-images FAILURE in 3m 51s
❌ cifmw-multinode-kuttl FAILURE in 18m 23s
❌ cifmw-tcib FAILURE in 3m 54s
❌ cifmw-architecture-validate-hci FAILURE in 3m 38s
❌ ci-framework-openstack-meta-content-provider FAILURE in 4m 13s

@michburk

michburk commented Jul 9, 2026

Copy link
Copy Markdown
Contributor Author

recheck

@centosinfra-prod-github-app

Copy link
Copy Markdown

Build failed (check pipeline). Post recheck (without leading slash)
to rerun all jobs. Make sure the failure cause has been resolved before
you rerun jobs.

https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/57754e3befc24996a786ee3acbbf0140

❌ openstack-k8s-operators-content-provider FAILURE in 4m 59s
⚠️ podified-multinode-edpm-deployment-crc SKIPPED Skipped due to failed job openstack-k8s-operators-content-provider
⚠️ cifmw-crc-podified-edpm-baremetal SKIPPED Skipped due to failed job openstack-k8s-operators-content-provider
⚠️ cifmw-crc-podified-edpm-baremetal-minor-update SKIPPED Skipped due to failed job openstack-k8s-operators-content-provider
⚠️ podified-multinode-hci-deployment-crc SKIPPED Skipped due to failed job openstack-k8s-operators-content-provider
⚠️ cifmw-multinode-tempest SKIPPED Skipped due to failed job openstack-k8s-operators-content-provider
⚠️ adoption-standalone-to-crc-ceph-provider SKIPPED Skipped due to failed job openstack-k8s-operators-content-provider
⚠️ whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp SKIPPED Skipped due to failed job openstack-k8s-operators-content-provider
✔️ noop SUCCESS in 0s
❌ cifmw-pod-ansible-test FAILURE in 7m 07s
✔️ cifmw-pod-pre-commit SUCCESS in 8m 59s
❌ cifmw-content-provider-build-images FAILURE in 4m 23s
❌ cifmw-multinode-kuttl FAILURE in 18m 47s
❌ cifmw-tcib FAILURE in 4m 12s
❌ cifmw-architecture-validate-hci FAILURE in 3m 59s
❌ ci-framework-openstack-meta-content-provider FAILURE in 4m 19s

@michburk

Copy link
Copy Markdown
Contributor Author

recheck

@centosinfra-prod-github-app

Copy link
Copy Markdown

Build failed (check pipeline). Post recheck (without leading slash)
to rerun all jobs. Make sure the failure cause has been resolved before
you rerun jobs.

https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/877800f7c868497b89ba887069303385

❌ openstack-k8s-operators-content-provider FAILURE in 4m 50s
⚠️ podified-multinode-edpm-deployment-crc SKIPPED Skipped due to failed job openstack-k8s-operators-content-provider
⚠️ cifmw-crc-podified-edpm-baremetal SKIPPED Skipped due to failed job openstack-k8s-operators-content-provider
⚠️ cifmw-crc-podified-edpm-baremetal-minor-update SKIPPED Skipped due to failed job openstack-k8s-operators-content-provider
⚠️ podified-multinode-hci-deployment-crc SKIPPED Skipped due to failed job openstack-k8s-operators-content-provider
⚠️ cifmw-multinode-tempest SKIPPED Skipped due to failed job openstack-k8s-operators-content-provider
⚠️ adoption-standalone-to-crc-ceph-provider SKIPPED Skipped due to failed job openstack-k8s-operators-content-provider
⚠️ whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp SKIPPED Skipped due to failed job openstack-k8s-operators-content-provider
✔️ noop SUCCESS in 0s
❌ cifmw-pod-ansible-test FAILURE in 7m 26s
✔️ cifmw-pod-pre-commit SUCCESS in 11m 20s
❌ cifmw-content-provider-build-images FAILURE in 4m 59s
❌ cifmw-multinode-kuttl FAILURE in 21m 43s
❌ cifmw-tcib FAILURE in 4m 27s
❌ cifmw-architecture-validate-hci FAILURE in 4m 49s
❌ ci-framework-openstack-meta-content-provider FAILURE in 6m 25s

@michburk

Copy link
Copy Markdown
Contributor Author

recheck

@centosinfra-prod-github-app

Copy link
Copy Markdown

Build failed (check pipeline). Post recheck (without leading slash)
to rerun all jobs. Make sure the failure cause has been resolved before
you rerun jobs.

https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/533f2e6b59f3488cb12c85b146978e17

❌ openstack-k8s-operators-content-provider FAILURE in 4m 11s
⚠️ podified-multinode-edpm-deployment-crc SKIPPED Skipped due to failed job openstack-k8s-operators-content-provider
⚠️ cifmw-crc-podified-edpm-baremetal SKIPPED Skipped due to failed job openstack-k8s-operators-content-provider
⚠️ cifmw-crc-podified-edpm-baremetal-minor-update SKIPPED Skipped due to failed job openstack-k8s-operators-content-provider
⚠️ podified-multinode-hci-deployment-crc SKIPPED Skipped due to failed job openstack-k8s-operators-content-provider
⚠️ cifmw-multinode-tempest SKIPPED Skipped due to failed job openstack-k8s-operators-content-provider
⚠️ adoption-standalone-to-crc-ceph-provider SKIPPED Skipped due to failed job openstack-k8s-operators-content-provider
⚠️ whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp SKIPPED Skipped due to failed job openstack-k8s-operators-content-provider
✔️ noop SUCCESS in 0s
✔️ cifmw-pod-ansible-test SUCCESS in 9m 18s
✔️ cifmw-pod-pre-commit SUCCESS in 8m 52s
❌ cifmw-content-provider-build-images FAILURE in 4m 02s
❌ cifmw-multinode-kuttl FAILURE in 20m 42s
❌ cifmw-tcib FAILURE in 4m 22s
❌ cifmw-architecture-validate-hci FAILURE in 3m 46s
❌ ci-framework-openstack-meta-content-provider FAILURE in 4m 10s

@michburk

Copy link
Copy Markdown
Contributor Author

recheck

@centosinfra-prod-github-app

Copy link
Copy Markdown

Build failed (check pipeline). Post recheck (without leading slash)
to rerun all jobs. Make sure the failure cause has been resolved before
you rerun jobs.

https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/7e086525c4d14698986a32476cb12a3f

❌ openstack-k8s-operators-content-provider FAILURE in 6m 35s
⚠️ podified-multinode-edpm-deployment-crc SKIPPED Skipped due to failed job openstack-k8s-operators-content-provider
⚠️ cifmw-crc-podified-edpm-baremetal SKIPPED Skipped due to failed job openstack-k8s-operators-content-provider
⚠️ cifmw-crc-podified-edpm-baremetal-minor-update SKIPPED Skipped due to failed job openstack-k8s-operators-content-provider
⚠️ podified-multinode-hci-deployment-crc SKIPPED Skipped due to failed job openstack-k8s-operators-content-provider
⚠️ cifmw-multinode-tempest SKIPPED Skipped due to failed job openstack-k8s-operators-content-provider
⚠️ adoption-standalone-to-crc-ceph-provider SKIPPED Skipped due to failed job openstack-k8s-operators-content-provider
⚠️ whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp SKIPPED Skipped due to failed job openstack-k8s-operators-content-provider
✔️ noop SUCCESS in 0s
❌ cifmw-pod-ansible-test FAILURE in 7m 55s
✔️ cifmw-pod-pre-commit SUCCESS in 10m 55s
❌ cifmw-content-provider-build-images FAILURE in 4m 16s
❌ cifmw-multinode-kuttl FAILURE in 21m 43s
❌ cifmw-tcib FAILURE in 4m 11s
❌ cifmw-architecture-validate-hci FAILURE in 3m 43s
❌ ci-framework-openstack-meta-content-provider FAILURE in 4m 42s

@michburk

Copy link
Copy Markdown
Contributor Author

recheck

1 similar comment
@michburk

Copy link
Copy Markdown
Contributor Author

recheck

@github-actions

github-actions Bot commented Aug 7, 2026

Copy link
Copy Markdown

This PR is stale because it has been for over 15 days with no activity.
Remove stale label or comment or this will be closed in 7 days.

@github-actions github-actions Bot added the Stale label Aug 7, 2026
@github-actions github-actions Bot closed this Aug 15, 2026
@michburk michburk reopened this Aug 31, 2026
@centosinfra-prod-github-app

Copy link
Copy Markdown

Build failed (check pipeline). Post recheck (without leading slash)
to rerun all jobs. Make sure the failure cause has been resolved before
you rerun jobs.

https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/6fea1af6f9ec47a3944121ca262d2ef8

✔️ openstack-k8s-operators-content-provider SUCCESS in 4h 02m 47s
✔️ podified-multinode-edpm-deployment-crc SUCCESS in 1h 32m 55s
✔️ cifmw-crc-podified-edpm-baremetal SUCCESS in 1h 43m 37s
✔️ cifmw-crc-podified-edpm-baremetal-minor-update SUCCESS in 2h 30m 33s
✔️ podified-multinode-hci-deployment-crc SUCCESS in 1h 47m 51s
✔️ cifmw-multinode-tempest SUCCESS in 2h 03m 24s
✔️ openstack-k8s-operators-content-provider-bootc SUCCESS in 3h 35m 13s
✔️ cifmw-crc-podified-edpm-baremetal-bootc SUCCESS in 1h 48m 31s
❌ adoption-standalone-to-crc-ceph-provider POST_FAILURE in 3h 08m 37s
✔️ whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp SUCCESS in 3h 48m 46s (non-voting)
✔️ noop SUCCESS in 0s
✔️ cifmw-pod-ansible-test SUCCESS in 9m 25s
✔️ cifmw-pod-pre-commit SUCCESS in 8m 30s
✔️ cifmw-content-provider-build-images SUCCESS in 38m 47s
✔️ cifmw-multinode-kuttl SUCCESS in 3h 08m 23s
✔️ cifmw-tcib SUCCESS in 39m 13s
✔️ cifmw-architecture-validate-hci SUCCESS in 6m 02s
✔️ ci-framework-openstack-meta-content-provider SUCCESS in 13m 58s

@github-actions github-actions Bot removed the Stale label Sep 1, 2026
@michburk

michburk commented Sep 3, 2026

Copy link
Copy Markdown
Contributor Author

recheck

@centosinfra-prod-github-app

Copy link
Copy Markdown

Build failed (check pipeline). Post recheck (without leading slash)
to rerun all jobs. Make sure the failure cause has been resolved before
you rerun jobs.

https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/cee5ecb4b549481f8ce300ba8619357c

✔️ openstack-k8s-operators-content-provider SUCCESS in 4h 06m 34s
✔️ podified-multinode-edpm-deployment-crc SUCCESS in 1h 36m 41s
❌ cifmw-crc-podified-edpm-baremetal FAILURE in 2h 20m 00s
✔️ cifmw-crc-podified-edpm-baremetal-minor-update SUCCESS in 2h 14m 59s
✔️ podified-multinode-hci-deployment-crc SUCCESS in 1h 51m 13s
❌ cifmw-multinode-tempest POST_FAILURE in 1h 57m 37s
✔️ openstack-k8s-operators-content-provider-bootc SUCCESS in 56m 54s
❌ cifmw-crc-podified-edpm-baremetal-bootc NODE_FAILURE Node(set) request 099-0000191662 failed in 0s
❌ adoption-standalone-to-crc-ceph-provider POST_FAILURE in 3h 07m 44s
✔️ whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp SUCCESS in 3h 41m 06s (non-voting)
✔️ noop SUCCESS in 0s
✔️ cifmw-pod-ansible-test SUCCESS in 9m 42s
✔️ cifmw-pod-pre-commit SUCCESS in 9m 11s
✔️ cifmw-content-provider-build-images SUCCESS in 40m 31s
✔️ cifmw-multinode-kuttl SUCCESS in 3h 02m 46s
❌ cifmw-tcib FAILURE in 20m 20s
✔️ cifmw-architecture-validate-hci SUCCESS in 5m 31s
✔️ ci-framework-openstack-meta-content-provider SUCCESS in 17m 53s

@michburk

Copy link
Copy Markdown
Contributor Author

recheck

@centosinfra-prod-github-app

Copy link
Copy Markdown

Build failed (check pipeline). Post recheck (without leading slash)
to rerun all jobs. Make sure the failure cause has been resolved before
you rerun jobs.

https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/3b55189ed54f4ace8ddd3c91a46bc7f6

✔️ openstack-k8s-operators-content-provider SUCCESS in 3h 56m 11s
✔️ podified-multinode-edpm-deployment-crc SUCCESS in 1h 24m 03s
✔️ cifmw-crc-podified-edpm-baremetal SUCCESS in 1h 34m 10s
✔️ cifmw-crc-podified-edpm-baremetal-minor-update SUCCESS in 1h 53m 31s
❌ podified-multinode-hci-deployment-crc FAILURE in 1h 40m 32s
❌ cifmw-multinode-tempest FAILURE in 1h 33m 04s
✔️ openstack-k8s-operators-content-provider-bootc SUCCESS in 2h 09m 08s
✔️ cifmw-crc-podified-edpm-baremetal-bootc SUCCESS in 1h 25m 46s
❌ adoption-standalone-to-crc-ceph-provider POST_FAILURE in 3h 05m 26s
❌ whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp FAILURE in 3h 41m 28s (non-voting)
✔️ noop SUCCESS in 0s
✔️ cifmw-pod-ansible-test SUCCESS in 8m 27s
✔️ cifmw-pod-pre-commit SUCCESS in 8m 14s
❌ cifmw-content-provider-build-images TIMED_OUT in 46m 06s
✔️ cifmw-multinode-kuttl SUCCESS in 3h 01m 17s
✔️ cifmw-tcib SUCCESS in 54m 28s
✔️ cifmw-architecture-validate-hci SUCCESS in 5m 29s
✔️ ci-framework-openstack-meta-content-provider SUCCESS in 13m 52s

@michburk

Copy link
Copy Markdown
Contributor Author

recheck

@centosinfra-prod-github-app

Copy link
Copy Markdown

Build failed (check pipeline). Post recheck (without leading slash)
to rerun all jobs. Make sure the failure cause has been resolved before
you rerun jobs.

https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/85bf87f49d504c6dbfb73d3d1b30faf4

✔️ openstack-k8s-operators-content-provider SUCCESS in 4h 14m 30s
✔️ podified-multinode-edpm-deployment-crc SUCCESS in 1h 25m 50s
❌ cifmw-crc-podified-edpm-baremetal NODE_FAILURE Node(set) request 099-0000207205 failed in 0s
✔️ cifmw-crc-podified-edpm-baremetal-minor-update SUCCESS in 2h 07m 35s
✔️ podified-multinode-hci-deployment-crc SUCCESS in 1h 39m 23s
❌ cifmw-multinode-tempest FAILURE in 1h 33m 26s
✔️ openstack-k8s-operators-content-provider-bootc SUCCESS in 1h 08m 33s
❌ cifmw-crc-podified-edpm-baremetal-bootc NODE_FAILURE Node(set) request 099-0000207240 failed in 0s
✔️ adoption-standalone-to-crc-ceph-provider SUCCESS in 3h 04m 52s
✔️ whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp SUCCESS in 3h 34m 10s (non-voting)
✔️ noop SUCCESS in 0s
✔️ cifmw-pod-ansible-test SUCCESS in 8m 41s
✔️ cifmw-pod-pre-commit SUCCESS in 8m 09s
✔️ cifmw-content-provider-build-images SUCCESS in 39m 23s
✔️ cifmw-multinode-kuttl SUCCESS in 2h 54m 14s
✔️ cifmw-tcib SUCCESS in 39m 52s
✔️ cifmw-architecture-validate-hci SUCCESS in 5m 53s
✔️ ci-framework-openstack-meta-content-provider SUCCESS in 13m 38s

@michburk

Copy link
Copy Markdown
Contributor Author

recheck

@centosinfra-prod-github-app

Copy link
Copy Markdown

Build failed (check pipeline). Post recheck (without leading slash)
to rerun all jobs. Make sure the failure cause has been resolved before
you rerun jobs.

https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/10499dd7d4544fa9b9e5e280e265cd62

✔️ openstack-k8s-operators-content-provider SUCCESS in 4h 13m 39s
✔️ podified-multinode-edpm-deployment-crc SUCCESS in 1h 21m 40s
✔️ cifmw-crc-podified-edpm-baremetal SUCCESS in 1h 41m 06s
✔️ cifmw-crc-podified-edpm-baremetal-minor-update SUCCESS in 1h 56m 01s
✔️ podified-multinode-hci-deployment-crc SUCCESS in 1h 38m 47s
❌ cifmw-multinode-tempest FAILURE in 1h 30m 14s
✔️ openstack-k8s-operators-content-provider-bootc SUCCESS in 2h 28m 43s
✔️ cifmw-crc-podified-edpm-baremetal-bootc SUCCESS in 1h 24m 45s
❌ adoption-standalone-to-crc-ceph-provider NODE_FAILURE Node(set) request 099-0000207378 failed in 0s
✔️ whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp SUCCESS in 3h 42m 49s (non-voting)
✔️ noop SUCCESS in 0s
✔️ cifmw-pod-ansible-test SUCCESS in 8m 29s
✔️ cifmw-pod-pre-commit SUCCESS in 7m 52s
✔️ cifmw-content-provider-build-images SUCCESS in 37m 01s
✔️ cifmw-multinode-kuttl SUCCESS in 2h 56m 12s
✔️ cifmw-tcib SUCCESS in 43m 14s
✔️ cifmw-architecture-validate-hci SUCCESS in 6m 00s
✔️ ci-framework-openstack-meta-content-provider SUCCESS in 13m 02s

@michburk

Copy link
Copy Markdown
Contributor Author

recheck

@centosinfra-prod-github-app

Copy link
Copy Markdown

Build failed (check pipeline). Post recheck (without leading slash)
to rerun all jobs. Make sure the failure cause has been resolved before
you rerun jobs.

https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/dc6dec5473ff44e3995ae1c29aa4657d

✔️ openstack-k8s-operators-content-provider SUCCESS in 3h 16m 33s
✔️ podified-multinode-edpm-deployment-crc SUCCESS in 1h 23m 35s
❌ cifmw-crc-podified-edpm-baremetal NODE_FAILURE Node(set) request 099-0000211661 failed in 0s
✔️ cifmw-crc-podified-edpm-baremetal-minor-update SUCCESS in 2h 01m 28s
✔️ podified-multinode-hci-deployment-crc SUCCESS in 1h 39m 46s
✔️ cifmw-multinode-tempest SUCCESS in 1h 42m 44s
❌ openstack-k8s-operators-content-provider-bootc FAILURE in 18m 44s
⚠️ cifmw-crc-podified-edpm-baremetal-bootc SKIPPED Skipped due to failed job openstack-k8s-operators-content-provider-bootc
✔️ adoption-standalone-to-crc-ceph-provider SUCCESS in 3h 02m 19s
❌ whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp FAILURE in 44m 50s (non-voting)
✔️ noop SUCCESS in 0s
✔️ cifmw-pod-ansible-test SUCCESS in 9m 10s
✔️ cifmw-pod-pre-commit SUCCESS in 8m 25s
❌ cifmw-content-provider-build-images FAILURE in 18m 13s
✔️ cifmw-multinode-kuttl SUCCESS in 2h 56m 46s
✔️ cifmw-tcib SUCCESS in 42m 40s
✔️ cifmw-architecture-validate-hci SUCCESS in 6m 00s
✔️ ci-framework-openstack-meta-content-provider SUCCESS in 13m 27s

@michburk

Copy link
Copy Markdown
Contributor Author

recheck

1 similar comment
@michburk

Copy link
Copy Markdown
Contributor Author

recheck

@centosinfra-prod-github-app

Copy link
Copy Markdown

Build failed (check pipeline). Post recheck (without leading slash)
to rerun all jobs. Make sure the failure cause has been resolved before
you rerun jobs.

https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/ebd4cbcbcd9b44f8a458cbaeb8676e0e

✔️ openstack-k8s-operators-content-provider SUCCESS in 5h 04m 53s
✔️ podified-multinode-edpm-deployment-crc SUCCESS in 1h 23m 49s
✔️ podified-multinode-edpm-deployment-crc-centos-10 SUCCESS in 1h 25m 04s
✔️ cifmw-crc-podified-edpm-baremetal SUCCESS in 1h 36m 59s
✔️ cifmw-crc-podified-edpm-baremetal-minor-update SUCCESS in 1h 58m 25s
✔️ podified-multinode-hci-deployment-crc SUCCESS in 1h 39m 41s
✔️ cifmw-multinode-tempest SUCCESS in 1h 42m 22s
✔️ openstack-k8s-operators-content-provider-bootc SUCCESS in 3h 41m 57s
✔️ cifmw-crc-podified-edpm-baremetal-bootc SUCCESS in 1h 27m 28s
❌ adoption-standalone-to-crc-ceph-provider POST_FAILURE in 3h 06m 37s
✔️ whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp SUCCESS in 3h 48m 09s (non-voting)
✔️ noop SUCCESS in 0s
✔️ cifmw-pod-ansible-test SUCCESS in 8m 29s
✔️ cifmw-pod-pre-commit SUCCESS in 8m 26s
✔️ cifmw-content-provider-build-images SUCCESS in 33m 43s
✔️ cifmw-multinode-kuttl SUCCESS in 3h 00m 33s
✔️ cifmw-tcib SUCCESS in 38m 49s
✔️ cifmw-architecture-validate-hci SUCCESS in 5m 34s
✔️ ci-framework-openstack-meta-content-provider SUCCESS in 12m 45s

@michburk

Copy link
Copy Markdown
Contributor Author

recheck

This reverts commit 9535333.

Signed-off-by: Michael Burke <michburk@redhat.com>
This reverts commit 1bc9f55.

Signed-off-by: Michael Burke <michburk@redhat.com>
@centosinfra-prod-github-app

Copy link
Copy Markdown

Build failed (check pipeline). Post recheck (without leading slash)
to rerun all jobs. Make sure the failure cause has been resolved before
you rerun jobs.

https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/2c7a9e3db0f442b48279335bc982eace

❌ openstack-k8s-operators-content-provider FAILURE in 5m 26s
⚠️ podified-multinode-edpm-deployment-crc SKIPPED Skipped due to failed job openstack-k8s-operators-content-provider
⚠️ podified-multinode-edpm-deployment-crc-centos-10 SKIPPED Skipped due to failed job openstack-k8s-operators-content-provider
⚠️ cifmw-crc-podified-edpm-baremetal SKIPPED Skipped due to failed job openstack-k8s-operators-content-provider
⚠️ cifmw-crc-podified-edpm-baremetal-minor-update SKIPPED Skipped due to failed job openstack-k8s-operators-content-provider
⚠️ podified-multinode-hci-deployment-crc SKIPPED Skipped due to failed job openstack-k8s-operators-content-provider
⚠️ cifmw-multinode-tempest SKIPPED Skipped due to failed job openstack-k8s-operators-content-provider
❌ openstack-k8s-operators-content-provider-bootc FAILURE in 5m 20s
⚠️ cifmw-crc-podified-edpm-baremetal-bootc SKIPPED Skipped due to failed job openstack-k8s-operators-content-provider-bootc
⚠️ adoption-standalone-to-crc-ceph-provider SKIPPED Skipped due to failed job openstack-k8s-operators-content-provider
⚠️ whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp SKIPPED Skipped due to failed job openstack-k8s-operators-content-provider (non-voting)
✔️ noop SUCCESS in 0s
✔️ cifmw-pod-ansible-test SUCCESS in 8m 57s
✔️ cifmw-pod-pre-commit SUCCESS in 8m 50s
❌ cifmw-content-provider-build-images FAILURE in 5m 29s
❌ cifmw-multinode-kuttl POST_FAILURE in 15m 44s
❌ cifmw-tcib FAILURE in 5m 03s
✔️ cifmw-architecture-validate-hci SUCCESS in 5m 36s
❌ ci-framework-openstack-meta-content-provider FAILURE in 5m 38s

This branch has not been deployed

No deployments
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant