Skip to content

[DNM] tobiko - testing s2i tobiko container image - #4113

Open
fyanac wants to merge 1 commit into
openstack-k8s-operators:mainfrom
fyanac:test_tobiko
Open

fyanac wants to merge 1 commit into
openstack-k8s-operators:mainfrom
fyanac:test_tobiko

Conversation

@fyanac

@fyanac fyanac commented Aug 17, 2026 •

Copy link
Copy Markdown
Contributor

tobiko - testing s2i tobiko container image
with podified-multinode-edpm-deployment-crc-test-operator-tobiko-s2i

Depends-On: #4218

@openshift-ci

openshift-ci Bot commented Aug 17, 2026

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by:
Once this PR has been reviewed and has the lgtm label, please assign nemarjan for approval. For more information see the Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@centosinfra-prod-github-app

Copy link
Copy Markdown

Build failed (check pipeline). Post recheck (without leading slash)
to rerun all jobs. Make sure the failure cause has been resolved before
you rerun jobs.

https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/9435cb388d1f4888a37fb53886938166

✔️ openstack-k8s-operators-content-provider SUCCESS in 2h 04m 32s
✔️ podified-multinode-edpm-deployment-crc SUCCESS in 1h 30m 09s
❌ cifmw-crc-podified-edpm-baremetal NODE_FAILURE Node(set) request 099-0000172628 failed in 0s
❌ cifmw-crc-podified-edpm-baremetal-minor-update NODE_FAILURE Node(set) request 099-0000172629 failed in 0s
✔️ podified-multinode-edpm-deployment-crc-test-operator-tobiko-s2i SUCCESS in 1h 32m 13s
✔️ cifmw-pod-zuul-files SUCCESS in 5m 00s
✔️ openstack-k8s-operators-content-provider-bootc SUCCESS in 41m 44s
❌ cifmw-crc-podified-edpm-baremetal-bootc NODE_FAILURE Node(set) request 099-0000172637 failed in 0s
✔️ noop SUCCESS in 0s
✔️ cifmw-pod-ansible-test SUCCESS in 8m 38s
✔️ cifmw-pod-pre-commit SUCCESS in 8m 49s

@centosinfra-prod-github-app

Copy link
Copy Markdown

Build failed (check pipeline). Post recheck (without leading slash)
to rerun all jobs. Make sure the failure cause has been resolved before
you rerun jobs.

https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/f799234c5def42d8be15075b4b3aea11

✔️ openstack-k8s-operators-content-provider SUCCESS in 3h 47m 38s
✔️ podified-multinode-edpm-deployment-crc-test-operator-tobiko-s2i SUCCESS in 1h 40m 56s
✔️ cifmw-pod-zuul-files SUCCESS in 5m 37s
✔️ openstack-k8s-operators-content-provider-bootc SUCCESS in 1h 18m 16s
❌ cifmw-crc-podified-edpm-baremetal-bootc NODE_FAILURE Node(set) request 099-0000173655 failed in 0s
✔️ noop SUCCESS in 0s
✔️ cifmw-pod-ansible-test SUCCESS in 9m 28s
✔️ cifmw-pod-pre-commit SUCCESS in 9m 04s

@fyanac

fyanac commented Aug 19, 2026

Copy link
Copy Markdown
Contributor Author

recheck

@fyanac

fyanac commented Aug 21, 2026

Copy link
Copy Markdown
Contributor Author

recheck

@centosinfra-prod-github-app

Copy link
Copy Markdown

Build failed (check pipeline). Post recheck (without leading slash)
to rerun all jobs. Make sure the failure cause has been resolved before
you rerun jobs.

https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/3836e7c4c0484ca084a7e7f70951f4f7

✔️ openstack-k8s-operators-content-provider SUCCESS in 2h 00m 15s
❌ podified-multinode-edpm-deployment-crc-test-operator-tobiko-s2i FAILURE in 1h 46m 01s
✔️ cifmw-pod-zuul-files SUCCESS in 5m 54s
✔️ openstack-k8s-operators-content-provider-bootc SUCCESS in 2h 11m 42s
✔️ cifmw-crc-podified-edpm-baremetal-bootc SUCCESS in 1h 38m 16s
✔️ noop SUCCESS in 0s
✔️ cifmw-pod-ansible-test SUCCESS in 10m 06s
✔️ cifmw-pod-pre-commit SUCCESS in 8m 59s

@fyanac

fyanac commented Aug 24, 2026

Copy link
Copy Markdown
Contributor Author

recheck

@centosinfra-prod-github-app

Copy link
Copy Markdown

Build failed (check pipeline). Post recheck (without leading slash)
to rerun all jobs. Make sure the failure cause has been resolved before
you rerun jobs.

https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/d4372e1ee1684e1dbeed0db048f5cb9e

✔️ openstack-k8s-operators-content-provider SUCCESS in 3h 01m 35s
❌ podified-multinode-edpm-deployment-crc-test-operator-tobiko-s2i FAILURE in 1h 48m 29s
✔️ cifmw-pod-zuul-files SUCCESS in 5m 35s
✔️ openstack-k8s-operators-content-provider-bootc SUCCESS in 3h 18m 34s
❌ cifmw-crc-podified-edpm-baremetal-bootc FAILURE in 2h 00m 12s
✔️ noop SUCCESS in 0s
✔️ cifmw-pod-ansible-test SUCCESS in 9m 44s
✔️ cifmw-pod-pre-commit SUCCESS in 9m 28s

@fyanac

fyanac commented Aug 25, 2026

Copy link
Copy Markdown
Contributor Author

recheck again the fix is merged -> related to "ModuleNotFoundError: No module named 'setuptools.extern.tomli'"

@fyanac

fyanac commented Aug 25, 2026

Copy link
Copy Markdown
Contributor Author

recheck

@centosinfra-prod-github-app

Copy link
Copy Markdown

Build failed (check pipeline). Post recheck (without leading slash)
to rerun all jobs. Make sure the failure cause has been resolved before
you rerun jobs.

https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/22b8f0acad8447c3ae74bca9fc8edce1

✔️ openstack-k8s-operators-content-provider SUCCESS in 2h 23m 46s
❌ podified-multinode-edpm-deployment-crc-test-operator-tobiko-s2i FAILURE in 1h 46m 08s
✔️ cifmw-pod-zuul-files SUCCESS in 5m 01s
✔️ openstack-k8s-operators-content-provider-bootc SUCCESS in 2h 26m 08s
✔️ cifmw-crc-podified-edpm-baremetal-bootc SUCCESS in 1h 49m 16s
✔️ noop SUCCESS in 0s
✔️ cifmw-pod-ansible-test SUCCESS in 9m 00s
✔️ cifmw-pod-pre-commit SUCCESS in 8m 40s

@fyanac

fyanac commented Aug 25, 2026

Copy link
Copy Markdown
Contributor Author

recheck

@fyanac

fyanac commented Aug 25, 2026

Copy link
Copy Markdown
Contributor Author

I've updated the tobiko image - rerun the job again.

@fyanac

fyanac commented Aug 25, 2026

Copy link
Copy Markdown
Contributor Author

recheck

@centosinfra-prod-github-app

Copy link
Copy Markdown

Build failed (check pipeline). Post recheck (without leading slash)
to rerun all jobs. Make sure the failure cause has been resolved before
you rerun jobs.

https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/bc88c29d731d42b689f213c9abdfc37d

✔️ openstack-k8s-operators-content-provider SUCCESS in 1h 42m 08s
❌ podified-multinode-edpm-deployment-crc-test-operator-tobiko-s2i FAILURE in 1h 18m 58s
✔️ cifmw-pod-zuul-files SUCCESS in 5m 45s
✔️ openstack-k8s-operators-content-provider-bootc SUCCESS in 3h 32m 46s
✔️ cifmw-crc-podified-edpm-baremetal-bootc SUCCESS in 1h 42m 24s
✔️ noop SUCCESS in 0s
✔️ cifmw-pod-ansible-test SUCCESS in 9m 13s
✔️ cifmw-pod-pre-commit SUCCESS in 8m 58s

@fyanac

fyanac commented Aug 25, 2026

Copy link
Copy Markdown
Contributor Author

recheck

@centosinfra-prod-github-app

Copy link
Copy Markdown

Build failed (check pipeline). Post recheck (without leading slash)
to rerun all jobs. Make sure the failure cause has been resolved before
you rerun jobs.

https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/8456ea24d27748049d2f40b99b80051e

✔️ openstack-k8s-operators-content-provider SUCCESS in 1h 46m 08s
❌ podified-multinode-edpm-deployment-crc-test-operator-tobiko-s2i FAILURE in 1h 21m 22s
✔️ cifmw-pod-zuul-files SUCCESS in 5m 27s
✔️ openstack-k8s-operators-content-provider-bootc SUCCESS in 2h 44m 02s
✔️ cifmw-crc-podified-edpm-baremetal-bootc SUCCESS in 1h 37m 21s
✔️ noop SUCCESS in 0s
✔️ cifmw-pod-ansible-test SUCCESS in 12m 33s
✔️ cifmw-pod-pre-commit SUCCESS in 10m 59s

@fyanac

fyanac commented Aug 26, 2026

Copy link
Copy Markdown
Contributor Author

recheck

@centosinfra-prod-github-app

Copy link
Copy Markdown

Build failed (check pipeline). Post recheck (without leading slash)
to rerun all jobs. Make sure the failure cause has been resolved before
you rerun jobs.

https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/85f967081e704917b0bfde190c6ee2ae

✔️ openstack-k8s-operators-content-provider SUCCESS in 1h 33m 51s
❌ podified-multinode-edpm-deployment-crc-test-operator-tobiko-s2i FAILURE in 1h 18m 52s
❌ cifmw-pod-zuul-files FAILURE in 5m 52s
✔️ noop SUCCESS in 0s
✔️ cifmw-pod-ansible-test SUCCESS in 9m 43s
✔️ cifmw-pod-pre-commit SUCCESS in 8m 34s

@fyanac

fyanac commented Aug 26, 2026

Copy link
Copy Markdown
Contributor Author

I've updated the tobiko image

@fyanac

fyanac commented Aug 26, 2026

Copy link
Copy Markdown
Contributor Author

recheck

@centosinfra-prod-github-app

Copy link
Copy Markdown

Build failed (check pipeline). Post recheck (without leading slash)
to rerun all jobs. Make sure the failure cause has been resolved before
you rerun jobs.

https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/535ad3f9a6e24dce8da13d35177c7e54

✔️ openstack-k8s-operators-content-provider SUCCESS in 2h 00m 44s
✔️ podified-multinode-edpm-deployment-crc-test-operator-tobiko-s2i SUCCESS in 1h 33m 31s
❌ cifmw-pod-zuul-files FAILURE in 7m 33s
✔️ noop SUCCESS in 0s
✔️ cifmw-pod-ansible-test SUCCESS in 11m 03s
✔️ cifmw-pod-pre-commit SUCCESS in 10m 34s

@fyanac

fyanac commented Aug 31, 2026

Copy link
Copy Markdown
Contributor Author

recheck

@centosinfra-prod-github-app

Copy link
Copy Markdown

Build failed (check pipeline). Post recheck (without leading slash)
to rerun all jobs. Make sure the failure cause has been resolved before
you rerun jobs.

https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/7aaa78214d174b058833af74ae365df7

✔️ openstack-k8s-operators-content-provider SUCCESS in 1h 36m 33s
❌ podified-multinode-edpm-deployment-crc-test-operator-tobiko-s2i FAILURE in 1h 19m 51s
❌ cifmw-pod-zuul-files FAILURE in 6m 59s
✔️ noop SUCCESS in 0s
✔️ cifmw-pod-ansible-test SUCCESS in 11m 13s
✔️ cifmw-pod-pre-commit SUCCESS in 10m 38s

@centosinfra-prod-github-app

Copy link
Copy Markdown

Build failed (check pipeline). Post recheck (without leading slash)
to rerun all jobs. Make sure the failure cause has been resolved before
you rerun jobs.

https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/fbf0dd5ed04d4f73aa405e9bb1bf7cb4

✔️ openstack-k8s-operators-content-provider SUCCESS in 2h 04m 18s
✔️ podified-multinode-edpm-deployment-crc-test-operator-tobiko-s2i SUCCESS in 1h 33m 29s
❌ cifmw-pod-zuul-files FAILURE in 5m 27s
✔️ noop SUCCESS in 0s
✔️ cifmw-pod-ansible-test SUCCESS in 10m 20s
✔️ cifmw-pod-pre-commit SUCCESS in 9m 15s

@fyanac

fyanac commented Sep 3, 2026

Copy link
Copy Markdown
Contributor Author

recheck

@centosinfra-prod-github-app

Copy link
Copy Markdown

Build failed (check pipeline). Post recheck (without leading slash)
to rerun all jobs. Make sure the failure cause has been resolved before
you rerun jobs.

https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/8c53019d0c69439aaff1562041fd08c0

✔️ openstack-k8s-operators-content-provider SUCCESS in 2h 03m 27s
❌ podified-multinode-edpm-deployment-crc-test-operator-tobiko-s2i FAILURE in 1h 28m 53s
❌ cifmw-pod-zuul-files FAILURE in 6m 05s
✔️ noop SUCCESS in 0s
✔️ cifmw-pod-ansible-test SUCCESS in 9m 24s
✔️ cifmw-pod-pre-commit SUCCESS in 9m 35s

@fyanac

fyanac commented Sep 4, 2026

Copy link
Copy Markdown
Contributor Author

recheck

@centosinfra-prod-github-app

Copy link
Copy Markdown

Build failed (check pipeline). Post recheck (without leading slash)
to rerun all jobs. Make sure the failure cause has been resolved before
you rerun jobs.

https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/bef3b471a33b42d59de45498d5f38515

✔️ openstack-k8s-operators-content-provider SUCCESS in 1h 26m 21s
❌ podified-multinode-edpm-deployment-crc-test-operator-tobiko-s2i FAILURE in 1h 10m 53s
❌ cifmw-pod-zuul-files FAILURE in 5m 34s
✔️ noop SUCCESS in 0s
✔️ cifmw-pod-ansible-test SUCCESS in 9m 29s
✔️ cifmw-pod-pre-commit SUCCESS in 9m 18s

@fyanac

fyanac commented Sep 4, 2026

Copy link
Copy Markdown
Contributor Author

recheck

@centosinfra-prod-github-app

Copy link
Copy Markdown

Build failed (check pipeline). Post recheck (without leading slash)
to rerun all jobs. Make sure the failure cause has been resolved before
you rerun jobs.

https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/fde6f07ef01d4f4d970106a88ec0144e

✔️ openstack-k8s-operators-content-provider SUCCESS in 18m 29s
❌ podified-multinode-edpm-deployment-crc-test-operator-tobiko-s2i NODE_FAILURE Node(set) request 099-0000192378 failed in 0s
❌ cifmw-pod-zuul-files FAILURE in 6m 39s
✔️ noop SUCCESS in 0s
✔️ cifmw-pod-ansible-test SUCCESS in 9m 18s
✔️ cifmw-pod-pre-commit SUCCESS in 10m 27s

@fyanac

fyanac commented Sep 7, 2026

Copy link
Copy Markdown
Contributor Author

recheck

@centosinfra-prod-github-app

Copy link
Copy Markdown

Build failed (check pipeline). Post recheck (without leading slash)
to rerun all jobs. Make sure the failure cause has been resolved before
you rerun jobs.

https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/749f009e1a7e4dcca39c1b9ccb8a7fe4

✔️ openstack-k8s-operators-content-provider SUCCESS in 14m 40s
❌ podified-multinode-edpm-deployment-crc-test-operator-tobiko-s2i NODE_FAILURE Node(set) request 099-0000194021 failed in 0s
❌ cifmw-pod-zuul-files FAILURE in 4m 46s
✔️ noop SUCCESS in 0s
✔️ cifmw-pod-ansible-test SUCCESS in 8m 26s
✔️ cifmw-pod-pre-commit SUCCESS in 8m 14s

@fyanac

fyanac commented Sep 7, 2026

Copy link
Copy Markdown
Contributor Author

recheck

@centosinfra-prod-github-app

Copy link
Copy Markdown

Build failed (check pipeline). Post recheck (without leading slash)
to rerun all jobs. Make sure the failure cause has been resolved before
you rerun jobs.

https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/032d41b8820848bf80af330eeded8df8

✔️ openstack-k8s-operators-content-provider SUCCESS in 13m 44s
❌ podified-multinode-edpm-deployment-crc-test-operator-tobiko-s2i NODE_FAILURE Node(set) request 099-0000194148 failed in 0s
❌ cifmw-pod-zuul-files FAILURE in 4m 24s
✔️ noop SUCCESS in 0s
✔️ cifmw-pod-ansible-test SUCCESS in 8m 24s
✔️ cifmw-pod-pre-commit SUCCESS in 8m 44s

@fyanac

fyanac commented Sep 9, 2026

Copy link
Copy Markdown
Contributor Author

recheck

@centosinfra-prod-github-app

Copy link
Copy Markdown

Build failed (check pipeline). Post recheck (without leading slash)
to rerun all jobs. Make sure the failure cause has been resolved before
you rerun jobs.

https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/864a5516d47643b5ab23d877125286d4

✔️ openstack-k8s-operators-content-provider SUCCESS in 1h 32m 12s
❌ podified-multinode-edpm-deployment-crc-test-operator-tobiko-s2i FAILURE in 50m 27s
❌ cifmw-pod-zuul-files FAILURE in 5m 03s
✔️ noop SUCCESS in 0s
✔️ cifmw-pod-ansible-test SUCCESS in 9m 21s
✔️ cifmw-pod-pre-commit SUCCESS in 8m 13s

@centosinfra-prod-github-app

Copy link
Copy Markdown

Build failed (check pipeline). Post recheck (without leading slash)
to rerun all jobs. Make sure the failure cause has been resolved before
you rerun jobs.

https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/8a1d9e14ed9c496e85be1a65c4895687

✔️ openstack-k8s-operators-content-provider SUCCESS in 1h 37m 26s
✔️ podified-multinode-edpm-deployment-crc-test-operator-tobiko-s2i SUCCESS in 1h 21m 51s
❌ cifmw-pod-zuul-files FAILURE in 5m 05s
✔️ noop SUCCESS in 0s
✔️ cifmw-pod-ansible-test SUCCESS in 8m 56s
✔️ cifmw-pod-pre-commit SUCCESS in 9m 28s

@fyanac

fyanac commented Sep 14, 2026

Copy link
Copy Markdown
Contributor Author

recheck

@centosinfra-prod-github-app

Copy link
Copy Markdown

Build failed (check pipeline). Post recheck (without leading slash)
to rerun all jobs. Make sure the failure cause has been resolved before
you rerun jobs.

https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/2bf1bfdeba784c19b86b116ac8001f25

❌ openstack-k8s-operators-content-provider FAILURE in 11m 09s
⚠️ podified-multinode-edpm-deployment-crc-test-operator-tobiko-s2i SKIPPED Skipped due to failed job openstack-k8s-operators-content-provider
❌ cifmw-pod-zuul-files FAILURE in 4m 37s
✔️ noop SUCCESS in 0s
✔️ cifmw-pod-ansible-test SUCCESS in 8m 52s
✔️ cifmw-pod-pre-commit SUCCESS in 8m 19s

@fyanac

fyanac commented Sep 15, 2026

Copy link
Copy Markdown
Contributor Author

recheck

@centosinfra-prod-github-app

Copy link
Copy Markdown

Build failed (check pipeline). Post recheck (without leading slash)
to rerun all jobs. Make sure the failure cause has been resolved before
you rerun jobs.

https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/b344ba9c32e0425a8f91763a8488b7ea

✔️ openstack-k8s-operators-content-provider SUCCESS in 1h 50m 27s
✔️ podified-multinode-edpm-deployment-crc-test-operator-tobiko-s2i SUCCESS in 1h 19m 01s
❌ cifmw-pod-zuul-files FAILURE in 4m 19s
✔️ noop SUCCESS in 0s
✔️ cifmw-pod-ansible-test SUCCESS in 8m 25s
✔️ cifmw-pod-pre-commit SUCCESS in 8m 14s

@fyanac

fyanac commented Sep 15, 2026

Copy link
Copy Markdown
Contributor Author

recheck

@centosinfra-prod-github-app

Copy link
Copy Markdown

Build failed (check pipeline). Post recheck (without leading slash)
to rerun all jobs. Make sure the failure cause has been resolved before
you rerun jobs.

https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/b9794e4a0fdc408ea3336fbce3f2d9f4

✔️ openstack-k8s-operators-content-provider SUCCESS in 1h 33m 36s
✔️ podified-multinode-edpm-deployment-crc-test-operator-tobiko-s2i SUCCESS in 1h 20m 11s
❌ cifmw-pod-zuul-files FAILURE in 4m 42s
✔️ noop SUCCESS in 0s
✔️ cifmw-pod-ansible-test SUCCESS in 8m 50s
✔️ cifmw-pod-pre-commit SUCCESS in 8m 58s

@fyanac

fyanac commented Sep 29, 2026

Copy link
Copy Markdown
Contributor Author

recheck

@centosinfra-prod-github-app

Copy link
Copy Markdown

Merge Failed.

This change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.
Warning:
Error merging github.com/openstack-k8s-operators/ci-framework for 4113,7788c627d0022d2a01553106c9bc6a1f66858dae

tobiko - testing s2i tobiko container image
with podified-multinode-edpm-deployment-crc-test-operator-tobiko-s2i
@centosinfra-prod-github-app

Copy link
Copy Markdown

Build failed (check pipeline). Post recheck (without leading slash)
to rerun all jobs. Make sure the failure cause has been resolved before
you rerun jobs.

https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/c62e551f1dbd4ed5820914a445c5d3b1

❌ openstack-k8s-operators-content-provider FAILURE in 6m 10s
⚠️ podified-multinode-edpm-deployment-crc-test-operator-tobiko-s2i SKIPPED Skipped due to failed job openstack-k8s-operators-content-provider
❌ cifmw-pod-zuul-files FAILURE in 4m 13s
✔️ noop SUCCESS in 0s
✔️ cifmw-pod-ansible-test SUCCESS in 8m 10s
✔️ cifmw-pod-pre-commit SUCCESS in 8m 34s

@fyanac

fyanac commented Sep 30, 2026

Copy link
Copy Markdown
Contributor Author

recheck

@centosinfra-prod-github-app

Copy link
Copy Markdown

Build failed (check pipeline). Post recheck (without leading slash)
to rerun all jobs. Make sure the failure cause has been resolved before
you rerun jobs.

https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/52682ab445f7482486d013cad447b577

❌ openstack-k8s-operators-content-provider FAILURE Job openstack-k8s-operators-content-provider requires artifact(s) cifmw-molecule provided by build ad508f1b04d949d283052b2f8e35ca52 (triggered by change 4218 on project openstack-k8s-operators/ci-framework), but that build failed with result "FAILURE"
⚠️ podified-multinode-edpm-deployment-crc-test-operator-tobiko-s2i SKIPPED Skipped due to failed job openstack-k8s-operators-content-provider
❌ cifmw-pod-zuul-files FAILURE in 4m 13s
✔️ noop SUCCESS in 0s
✔️ cifmw-pod-ansible-test SUCCESS in 8m 18s
✔️ cifmw-pod-pre-commit SUCCESS in 8m 35s

This branch has not been deployed

No deployments
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant