Skip to content

chore(deps): bump the go-k8s-dependencies group across 1 directory with 3 updates - #3300

Merged
tekton-robot merged 1 commit into
mainfrom
dependabot/go_modules/go-k8s-dependencies-002c455c37
Oct 6, 2026
Merged

tekton-robot merged 1 commit into
mainfrom
dependabot/go_modules/go-k8s-dependencies-002c455c37

Conversation

@dependabot

@dependabot dependabot Bot commented on behalf of github Sep 29, 2026 •

Copy link
Copy Markdown
Contributor

Bumps the go-k8s-dependencies group with 3 updates in the / directory: k8s.io/api, k8s.io/apimachinery and k8s.io/client-go.

Updates k8s.io/api from 0.36.4 to 0.36.5

Commits

Updates k8s.io/apimachinery from 0.36.4 to 0.36.5

Commits

Updates k8s.io/client-go from 0.36.4 to 0.36.5

Commits

@dependabot dependabot Bot added dependencies Used by dependabot - identifies all PRs created by dependabot kind/misc Categorizes issue or PR as a miscellaneuous one. ok-to-test Indicates a non-member PR verified by an org member that is safe to test. release-note-none Denotes a PR that doesnt merit a release note. labels Sep 29, 2026
@tekton-robot tekton-robot added the size/S Denotes a PR that changes 10-29 lines, ignoring generated files. label Sep 29, 2026
@dependabot dependabot Bot changed the title chore(deps): bump the go-k8s-dependencies group with 3 updates chore(deps): bump the go-k8s-dependencies group across 1 directory with 3 updates Oct 5, 2026
@dependabot
dependabot Bot force-pushed the dependabot/go_modules/go-k8s-dependencies-002c455c37 branch from 32ee192 to 76bb3a9 Compare October 5, 2026 08:56
…th 3 updates

Bumps the go-k8s-dependencies group with 3 updates in the / directory: [k8s.io/api](https://github.com/kubernetes/api), [k8s.io/apimachinery](https://github.com/kubernetes/apimachinery) and [k8s.io/client-go](https://github.com/kubernetes/client-go).


Updates `k8s.io/api` from 0.36.4 to 0.36.5
- [Commits](kubernetes/api@v0.36.4...v0.36.5)

Updates `k8s.io/apimachinery` from 0.36.4 to 0.36.5
- [Commits](kubernetes/apimachinery@v0.36.4...v0.36.5)

Updates `k8s.io/client-go` from 0.36.4 to 0.36.5
- [Changelog](https://github.com/kubernetes/client-go/blob/master/CHANGELOG.md)
- [Commits](kubernetes/client-go@v0.36.4...v0.36.5)

---
updated-dependencies:
- dependency-name: k8s.io/api
  dependency-version: 0.36.5
  dependency-type: direct:production
  update-type: version-update:semver-patch
  dependency-group: go-k8s-dependencies
- dependency-name: k8s.io/apimachinery
  dependency-version: 0.36.5
  dependency-type: direct:production
  update-type: version-update:semver-patch
  dependency-group: go-k8s-dependencies
- dependency-name: k8s.io/client-go
  dependency-version: 0.36.5
  dependency-type: direct:production
  update-type: version-update:semver-patch
  dependency-group: go-k8s-dependencies
...

Signed-off-by: dependabot[bot] <support@github.com>
@dependabot
dependabot Bot force-pushed the dependabot/go_modules/go-k8s-dependencies-002c455c37 branch from 76bb3a9 to a90aa34 Compare October 6, 2026 05:02

@divyansh42 divyansh42 left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

/lgtm

@tekton-robot tekton-robot added the lgtm Indicates that a PR is ready to be merged. label Oct 6, 2026
@tekton-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: divyansh42

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@tekton-robot tekton-robot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Oct 6, 2026
@tekton-robot
tekton-robot merged commit dffe58b into main Oct 6, 2026
12 checks passed
@dependabot
dependabot Bot deleted the dependabot/go_modules/go-k8s-dependencies-002c455c37 branch October 6, 2026 05:51
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. dependencies Used by dependabot - identifies all PRs created by dependabot kind/misc Categorizes issue or PR as a miscellaneuous one. lgtm Indicates that a PR is ready to be merged. ok-to-test Indicates a non-member PR verified by an org member that is safe to test. release-note-none Denotes a PR that doesnt merit a release note. size/S Denotes a PR that changes 10-29 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants